Skip to content

chore(tailwind): use @theme inline and add docs - #5319

Open
Barsnes wants to merge 2 commits into
mainfrom
chore/tailwind-docs
Open

Barsnes wants to merge 2 commits into
mainfrom
chore/tailwind-docs

Conversation

@Barsnes

@Barsnes Barsnes commented Sep 8, 2026

Copy link
Copy Markdown
Member

Summary

resolves #4317

This was first generated by Claude, then modified and check by me.

Checks

@changeset-bot

changeset-bot Bot commented Sep 8, 2026

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 81b8221

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 5 packages
Name Type
@digdir/designsystemet Patch
@digdir/designsystemet-css Patch
@digdir/designsystemet-react Patch
@digdir/designsystemet-types Patch
@digdir/designsystemet-web Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@github-actions

github-actions Bot commented Sep 8, 2026

Copy link
Copy Markdown
Contributor

Preview deployments for this pull request:

storybook - 17. Sep 2026 - 09:41

www - 17. Sep 2026 - 09:42

@Barsnes
Barsnes marked this pull request as ready for review September 8, 2026 09:15
@eirikbacker

Copy link
Copy Markdown
Contributor

Fantastic PR! 🤗 Only one though; Tailwind v3 supports @theme but not @theme inline - should we have a config --tailwind=v3 or similar as well for backwards compatible export? 🤔

@Barsnes

Barsnes commented Sep 8, 2026

Copy link
Copy Markdown
Member Author

Fantastic PR! 🤗 Only one though; Tailwind v3 supports @theme but not @theme inline - should we have a config --tailwind=v3 or similar as well for backwards compatible export? 🤔

@mimarz needs to decide this

@mimarz

mimarz commented Sep 10, 2026

Copy link
Copy Markdown
Collaborator

Fantastic PR! 🤗 Only one though; Tailwind v3 supports @theme but not @theme inline - should we have a config --tailwind=v3 or similar as well for backwards compatible export? 🤔

@mimarz needs to decide this

I agreed with the config option, but lets look at it next week how we expose/enable this :)

@mimarz

mimarz commented Sep 17, 2026

Copy link
Copy Markdown
Collaborator

I will make a new PR where we enable the new output config field. When doing so I can pull tailwind out of experimental there, since building the current way (using tokens build) will be deprecated.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

Use @theme inline in tailwind.css

3 participants