Skip to content

Add CODEOWNERS for automatic review assignment - #16447

Closed
diox wants to merge 1 commit into
mozilla:mainfrom
diox:codeowners
Closed

Add CODEOWNERS for automatic review assignment#16447
diox wants to merge 1 commit into
mozilla:mainfrom
diox:codeowners

Conversation

@diox

@diox diox commented Sep 11, 2026

Copy link
Copy Markdown
Member

No description provided.

@diox diox closed this Sep 11, 2026
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant