Skip to content

Add Nexus to CODEOWNERS - #798

Open
atol wants to merge 1 commit into
mainfrom
alicelin/nexus-codeowners
Open

Add Nexus to CODEOWNERS#798
atol wants to merge 1 commit into
mainfrom
alicelin/nexus-codeowners

Conversation

@atol

@atol atol commented Sep 9, 2026

Copy link
Copy Markdown
Contributor

What was changed

Add the Nexus team as codeowners for samples and tests that start with "nexus". Also, add the Nexus team as co-owners of the README and build.gradle.

@atol
atol marked this pull request as ready for review September 9, 2026 17:01
@atol
atol requested review from a team, antmendoza and tsurdilo as code owners September 9, 2026 17:01
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant